[WIP] 🐛 Implement support for multiNamespaceInformer in controller util consistency store - #14283
sbueringer wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
0ef6b8f to
e34ff2e
Compare
|
/cherry-pick release-1.14 (just as a reminder, will require a manual cherry-pick) |
|
@sbueringer: once the present PR merges, I will cherry-pick it on top of DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/test pull-cluster-api-e2e-main-gke |
What this PR does / why we need it:
Just a prototype atm.
Which issue(s) this PR fixes (optional, in
fixes #<issue number>(, fixes #<issue_number>, ...)format, will close the issue(s) when PR gets merged):Fixes #14269