Skip to content

pkg: bump noobaa version to get latest fixes - #423

Merged
openshift-merge-bot[bot] merged 1 commit into
red-hat-storage:mainfrom
subhamkrai:fix/DFBUGS-11485-bump-noobaa-operator-pin
Oct 6, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
red-hat-storage:mainfrom
subhamkrai:fix/DFBUGS-11485-bump-noobaa-operator-pin

Conversation

@subhamkrai

Copy link
Copy Markdown
Contributor

No description provided.

Signed-off-by: subhamkrai <srai@redhat.com>
@openshift-ci openshift-ci Bot added the approved label Oct 6, 2026
@subhamkrai
subhamkrai requested a review from travisn October 6, 2026 09:33

@travisn travisn left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Oct 6, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: subhamkrai, travisn

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 888f7a9 into red-hat-storage:main Oct 6, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants